Repository navigation
Feature/remap group ids - #1402
Open
BishopCodes wants to merge 3 commits into
Open
BishopCodes wants to merge 3 commits into
BishopCodes wants to merge 3 commits into
Conversation
…ns that do not follow the expected convention
🦋 Changeset detectedLatest commit: 85ed136 The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
This branch has not been deployed
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This change introduces the capabilities to remap the one trust group ids in the case a given setup does not conform to what Segment expects. The example case is that the ids are 1,2,3,4,5 which map to necessary, performance, functional, targeting, and other. Since this wrapper initializes higher up in the call stack using middleware or plugins on the analytics side does not work correctly. Additionally, the Segment UI does not support the consent key to start with a number.
I have added two unit tests one being in the mock wrapper and the other on the getNormalizedCategories function directly, in which all pass.
This will be ideal to have available to avoid having to make a custom wrapper for this given situation.
yarn changeset. Read about changesets here).