Skip to content

Translated button title - #1028

Closed
nijkor wants to merge 1 commit into
reactjs:mainfrom
nijkor:patch-5
Closed

nijkor wants to merge 1 commit into
reactjs:mainfrom
nijkor:patch-5

Conversation

@nijkor

@nijkor nijkor commented Jul 8, 2025 •

Copy link
Copy Markdown
Contributor

добавил перевод кнопке

@vercel

vercel Bot commented Jul 8, 2025 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

1 Skipped Deployment
Name Status Preview Comments Updated (UTC)
ru-legacy-reactjs-org ⬜️ Ignored (Inspect) Visit Preview Jul 8, 2025 2:17am

This branch was successfully deployed

1 active deployment
Preview – ru-react-dev — f80ceee9 Deployed Jul 8, 2025 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant