[pre-commit.ci] pre-commit autoupdate - #667
Conversation
|
Build succeeded. ✔️ pre-commit SUCCESS in 1m 37s |
4966a2c to
a094692
Compare
|
Build succeeded. ✔️ pre-commit SUCCESS in 1m 33s |
a094692 to
3e4e414
Compare
|
Build succeeded. ✔️ pre-commit SUCCESS in 1m 30s |
3e4e414 to
fc87a18
Compare
|
Build succeeded. ✔️ pre-commit SUCCESS in 1m 32s |
fc87a18 to
b88b38b
Compare
|
Build succeeded. ✔️ pre-commit SUCCESS in 1m 29s |
b88b38b to
79b464d
Compare
|
Build succeeded. ✔️ pre-commit SUCCESS in 1m 37s |
177157c to
3ec9aa8
Compare
|
❌ pre-commit FAILURE in 1m 35s |
3ec9aa8 to
0a07c93
Compare
|
❌ pre-commit FAILURE in 1m 37s |
|
Build succeeded. ✔️ pre-commit SUCCESS in 1m 31s |
839fc35 to
d48d61e
Compare
|
Build succeeded. ✔️ pre-commit SUCCESS in 1m 41s |
fbbbc40 to
23de7fd
Compare
|
❌ pre-commit FAILURE in 1m 34s |
updates: - [github.com/pre-commit/pre-commit-hooks: v5.0.0 → v6.0.0](pre-commit/pre-commit-hooks@v5.0.0...v6.0.0) - [github.com/ansible/ansible-lint.git: v25.7.0 → v26.8.0](https://github.com/ansible/ansible-lint.git/compare/v25.7.0...v26.8.0) - https://github.com/psf/black → https://github.com/psf/black-pre-commit-mirror - [github.com/psf/black-pre-commit-mirror: 25.1.0 → 26.5.1](psf/black-pre-commit-mirror@25.1.0...26.5.1) - [github.com/shellcheck-py/shellcheck-py: v0.10.0.1 → v0.11.0.1-1](shellcheck-py/shellcheck-py@v0.10.0.1...v0.11.0.1-1)
a1a06e1 to
572a6df
Compare
|
❌ pre-commit FAILURE in 1m 38s |
for more information, see https://pre-commit.ci
|
❌ pre-commit FAILURE in 1m 46s |
updates: