Skip to content

Add CODEOWNERS file - #128

Open
Jibola wants to merge 1 commit into
mainfrom
add-codeowners
Open

Add CODEOWNERS file#128
Jibola wants to merge 1 commit into
mainfrom
add-codeowners

Conversation

@Jibola

@Jibola Jibola commented Sep 4, 2026

Copy link
Copy Markdown
Collaborator

Adds a .github/CODEOWNERS file and adds @mongodb-labs/dbx-drivers-tools-maintainers.

Assign default ownership of the repository to
@mongodb-labs/dbx-drivers-tools-maintainers.

@blink1073 blink1073 left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants