Skip to content

Bump golangci-lint - #3106

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
shiftstack:bump-golangci-lint
Sep 4, 2026
Merged

Bump golangci-lint#3106
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
shiftstack:bump-golangci-lint

Conversation

@stephenfin

Copy link
Copy Markdown
Member

What this PR does / why we need it:

Bump golangci-lint. There are no no new linter violations and this should help minimise future breakages as we e.g. bump go versions.

Which issue this PR fixes(if applicable):

n/a

Special notes for reviewers:

n/a

Release note:

NONE

@k8s-ci-robot k8s-ci-robot added release-note-none Denotes a PR that doesn't merit a release note. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels May 2, 2026
@k8s-ci-robot
k8s-ci-robot requested review from kayrus and zetaab May 2, 2026 12:24
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign stephenfin for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label May 2, 2026
@stephenfin

Copy link
Copy Markdown
Member Author

/assign @kayrus

@kayrus

kayrus commented Jun 22, 2026

Copy link
Copy Markdown
Contributor

/lgtm
/approve

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jun 22, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: kayrus

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 22, 2026
@stephenfin

Copy link
Copy Markdown
Member Author

/retest-required

No new linter violations, but this should help minimise future breakages
as we e.g. bump go versions.

Signed-off-by: Stephen Finucane <stephenfin@redhat.com>
@kubernetes-prow kubernetes-prow Bot removed the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Sep 3, 2026

@mandre mandre left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Sep 4, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit cdf1cfe into kubernetes:master Sep 4, 2026
4 checks passed
@stephenfin

Copy link
Copy Markdown
Member Author

/cherry-pick release-1.36

I think a recent image bump in test-infra may have broken branches using the older version release-1.36

@k8s-infra-cherrypick-robot

Copy link
Copy Markdown

@stephenfin: new pull request created: #3198

Details

In response to this:

/cherry-pick release-1.36

I think a recent image bump in test-infra may have broken branches using the older version release-1.36

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@stephenfin

Copy link
Copy Markdown
Member Author

/cherry-pick release-1.35 release-1.34

We also need to fix these branches

@k8s-infra-cherrypick-robot

Copy link
Copy Markdown

@stephenfin: new pull request created: #3201

Details

In response to this:

/cherry-pick release-1.35 release-1.34

We also need to fix these branches

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. release-note-none Denotes a PR that doesn't merit a release note. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

7 participants